Repository navigation
fix(migrate): add the index module import only if the file exists - #1395
Merged
Merged
Conversation
|
Install the latest version of pnx https://pkg.svelte.dev/sv/c/60b61dc415536a3b7888b5646d239a1da36c843b createOpen in Note This PR is from a fork. A maintainer must approve approve each commit before it can be built and installed. |
🦋 Changeset detectedLatest commit: 60b61dc The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
sacrosanctic
force-pushed
the
lib-alias-index
branch
from
October 3, 2026 22:36
ac6555b to
0db3ddf
Compare
Closes # ### Description I wrote it, not sure if we should do it. ### Checklist - Update [snapshots](../CONTRIBUTING.md#update-snapshots) (if applicable) - Add a [changeset](../CONTRIBUTING.md#generating-changelogs) (if applicable) - Allow maintainers to edit this PR - I care about what I'm doing, no matter the tool I use (Notepad, Sublime, VSCode, AI...)
sacrosanctic
force-pushed
the
lib-alias-index
branch
from
October 3, 2026 22:39
0db3ddf to
7988bc4
Compare
sacrosanctic
marked this pull request as ready for review
October 4, 2026 12:52
…pulation to robustify/simplify it
dummdidumm
approved these changes
Oct 5, 2026
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Closes #
Related to #1392 and #1411
Description
I wrote it, not sure if we should do it.
Checklist