Skip to content

Add feature_flags field to PutStoreRequest and StoreHeartbeatResponse - #1396

Draft
MyonKeminta wants to merge 2 commits into
pingcap:masterfrom
MyonKeminta:m/feature-flags
Draft

MyonKeminta wants to merge 2 commits into
pingcap:masterfrom
MyonKeminta:m/feature-flags

Conversation

@MyonKeminta

@MyonKeminta MyonKeminta commented Jan 12, 2026 •

Copy link
Copy Markdown
Contributor

This helps implementing feature gate without depending a valid incremental semver.
Ref https://github.com/tidbcloud/cloud-storage-engine/issues/4052
Ref tikv/pd#10204

Signed-off-by: MyonKeminta <MyonKeminta@users.noreply.github.com>
@ti-chi-bot ti-chi-bot Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jan 12, 2026
Signed-off-by: MyonKeminta <MyonKeminta@users.noreply.github.com>
@ti-chi-bot

ti-chi-bot Bot commented Feb 3, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign nolouch, zhangjinpeng87 for approval. For more information see the Code Review Process.
Please ensure that each of them provides their approval before proceeding.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant