Skip to content

Mark old GC APIs and messages as deprecated - #1333

Open
MyonKeminta wants to merge 3 commits into
pingcap:masterfrom
MyonKeminta:m/deprecate-old-gc-api
Open

MyonKeminta wants to merge 3 commits into
pingcap:masterfrom
MyonKeminta:m/deprecate-old-gc-api

Conversation

@MyonKeminta

@MyonKeminta MyonKeminta commented Jul 18, 2025 •

Copy link
Copy Markdown
Contributor

ref: tikv/pd#8978

If this update causes your repo reporting compilation warnings or errors complaining about using deprecated code, it means your repo still has code using some obsolete APIs. For these code:

  • If it only needs to support next gen or v9.0+, consider migrate to using the new APIs
    • GetGCSafePoint(V2), UpdateGCSafePoint(V2) -> GetGCStates, GetAllKeyspacesGCStates
    • UpdateServiceGCSafePoint(V2) -> SetGCBarrier, DeleteGCBarrier, SetGlobalGCBarrier, DeleteGlobalGCBarrier
  • If it needs to support clusters in old versions (<v9.0 and non-nextgen), then you'll still need to use these old APIs. In these cases, please supress the warnings using //nolint:staticcheck (for golang) or #[allow(deprecated)] (for rust),

Signed-off-by: MyonKeminta <MyonKeminta@users.noreply.github.com>
@ti-chi-bot
ti-chi-bot Bot requested review from pingyu and yibin87 July 18, 2025 09:47
@ti-chi-bot ti-chi-bot Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Jul 18, 2025
@MyonKeminta MyonKeminta added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 18, 2025
@ti-chi-bot

ti-chi-bot Bot commented Nov 4, 2025

Copy link
Copy Markdown

@JaySon-Huang: adding LGTM is restricted to approvers and reviewers in OWNERS files.

Details

In response to this:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ti-chi-bot

ti-chi-bot Bot commented Nov 5, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: JaySon-Huang, rleungx

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot

ti-chi-bot Bot commented Nov 5, 2025

Copy link
Copy Markdown

[LGTM Timeline notifier]

Timeline:

  • 2025-11-05 06:18:25.777192903 +0000 UTC m=+250955.220222772: ☑️ agreed by rleungx.

@ti-chi-bot ti-chi-bot Bot added needs-1-more-lgtm Indicates a PR needs 1 more LGTM. approved labels Nov 5, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. needs-1-more-lgtm Indicates a PR needs 1 more LGTM. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants