Skip to content

OpenAPI schema sync (v3.9.0) - #295

Merged
mcnulty-fp merged 4 commits into
mainfrom
feat/open-api-v3.9.0
Oct 9, 2026
Merged

mcnulty-fp merged 4 commits into
mainfrom
feat/open-api-v3.9.0

Conversation

@fingerprint-dx-team-actions-runner

@fingerprint-dx-team-actions-runner fingerprint-dx-team-actions-runner Bot commented Oct 8, 2026 •

Copy link
Copy Markdown
Contributor

Schema sync for v3.9.0 OpenAPI schema release.

Ticket: INTER-2634

@changeset-bot

changeset-bot Bot commented Oct 8, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ba11769

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Oct 8, 2026 •

Copy link
Copy Markdown
Contributor

Coverage report

St.❔
Category Percentage Covered / Total
🟢 Statements 98.25% 168/171
🟢 Branches 97.03% 98/101
🟢 Functions 100% 37/37
🟢 Lines 98.21% 165/168

Test suite run success

132 tests passing in 30 suites.

Report generated by 🧪jest coverage report action from ba11769

Show full coverage report
St File % Stmts % Branch % Funcs % Lines Uncovered Line #s
🟢 All files 98.24 97.02 100 98.21
🟢  src 98.54 98.76 100 98.51
🔴   ...edApiTypes.ts 0 0 0 0
🔴   index.ts 0 0 0 0
🟢   sealedResults.ts 100 100 100 100
🟢   ...rApiClient.ts 96.15 97.22 100 96.15 360,364
🟢   types.ts 100 100 100 100
🟢   urlUtils.ts 100 100 100 100
🟢   webhook.ts 100 100 100 100
🟢  src/errors 97.05 90 100 96.96
🟢   apiErrors.ts 100 100 100 100
🟢   ...orResponse.ts 100 100 100 100
🟢   toError.ts 87.5 88.88 100 87.5 21
🟢   unsealError.ts 100 50 100 100 14

@github-actions

github-actions Bot commented Oct 9, 2026

Copy link
Copy Markdown
Contributor

🚀 Following releases will be created using changesets from this PR:

@fingerprint/node-sdk@7.9.0-test.0

Minor Changes

  • events-search: The 504 response for the Events Search endpoint is deprecated and will no longer be generated by the Server API. (ba11769)
  • events: The 504 response for the Get Event endpoint is deprecated and will no longer be generated by the Server API. (ba11769)
  • events-search: Add 503 Service Temporarily Unavailable response to the Events Search endpoint (aa5b2fc)
  • events: Add 503 Service Temporarily Unavailable response to the Get Event endpoint (aa5b2fc)

@mcnulty-fp
mcnulty-fp merged commit 3d46951 into main Oct 9, 2026
19 checks passed
@mcnulty-fp
mcnulty-fp deleted the feat/open-api-v3.9.0 branch October 9, 2026 17:05
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants