docs: update cover image in readme - #13
Conversation
|
Important Review skippedReview was skipped due to path filters ⛔ Files ignored due to path filters (1)
CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including You can disable this status message by setting the Comment |
|
📦 Python package built successfully!
|
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## filipzitny/mar-340-create-deepnote-toolkit-readme #13 +/- ##
=====================================================================================
+ Coverage 76.58% 76.60% +0.02%
=====================================================================================
Files 98 99 +1
Lines 5471 5476 +5
Branches 748 748
=====================================================================================
+ Hits 4190 4195 +5
Misses 1281 1281 ☔ View full report in Codecov by Sentry. |
7630f15
into
filipzitny/mar-340-create-deepnote-toolkit-readme
|
🚀 Review App Deployment Started
|
No description provided.