Skip to content

fix: improve command handling, ensure env is loaded, improve tui ux - #18

Merged
aid7n merged 2 commits into
mainfrom
chore/stabilize
Jul 17, 2026
Merged

aid7n merged 2 commits into
mainfrom
chore/stabilize

Conversation

@aid7n

@aid7n aid7n commented Jul 17, 2026

Copy link
Copy Markdown
Owner

No description provided.

@aid7n aid7n self-assigned this Jul 17, 2026
@changeset-bot

changeset-bot Bot commented Jul 17, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a478cbf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 9 packages
Name Type
@7x/jb-core Patch
@7x/jb Patch
@7x/jb-config Patch
@7x/jb-darwin-arm64 Patch
@7x/jb-darwin-x64 Patch
@7x/jb-linux-arm64 Patch
@7x/jb-linux-x64 Patch
@7x/jb-win32-arm64 Patch
@7x/jb-win32-x64 Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@aid7n aid7n added bug Something isn't working documentation Improvements or additions to documentation enhancement New feature or request dependencies Pull requests that update a dependency file labels Jul 17, 2026
@aid7n
aid7n merged commit 65548c9 into main Jul 17, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

bug Something isn't working dependencies Pull requests that update a dependency file documentation Improvements or additions to documentation enhancement New feature or request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant