Skip to content

chores(extracts): try blocks cleaned + pipeline update + reporting R … - #121

Merged
EstebanMontandon merged 1 commit into
mainfrom
SNT25-659-extraction
Sep 16, 2026
Merged

EstebanMontandon merged 1 commit into
mainfrom
SNT25-659-extraction

Conversation

@EstebanMontandon

Copy link
Copy Markdown
Collaborator

Fourth (not in regular SNT workflow order) cleaning:

Scripts touched:

Report:
snt_dhis2_extracts_report.ipynb

.R files touched:
snt_dhis2_extracts_report_report.r

Re-arranged try: block in the python pipeline snt_dhis2_extracts_report (no major functional changes, only error handling behavior)

…standard + readme + pyproject ruff rules updated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Hwy @EstebanMontandon ! Did you produce this update to the readme.md using the https://github.com/BLSQ/snt_development/blob/main/docs/PIPELINE_README_STANDARD.md (via agent)? If yes that would be great because it means it works :)

@sPuntinG sPuntinG left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Tested by uploading the changed files to SNT Testing + manually replacing the pipeline.py code of pipeline with the ./snt_dhis2_extract/pipeline.py code from this branch.
Pipeline run without errors.
Only tested for BFA.

@EstebanMontandon
EstebanMontandon merged commit fbc1910 into main Sep 16, 2026
@EstebanMontandon
EstebanMontandon deleted the SNT25-659-extraction branch September 16, 2026 14:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants